Skip to content

Update CODEOWNERS to include new members - #530

Merged
Mike Harder (mikeharder) merged 1 commit into
mainfrom
mikeharder/codeowners
Sep 16, 2026
Merged

Mike Harder (mikeharder) merged 1 commit into
mainfrom
mikeharder/codeowners

Conversation

@mikeharder

Copy link
Copy Markdown
Contributor

No description provided.

@azure-pipelines

Copy link
Copy Markdown
Azure Pipelines:
Successfully started running 1 pipeline(s).
There may be pipelines that require an authorized user to comment /azp run to run.

@mikeharder
Mike Harder (mikeharder) merged commit 1338484 into main Sep 16, 2026
8 checks passed
@mikeharder
Mike Harder (mikeharder) deleted the mikeharder/codeowners branch September 16, 2026 17:35
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants