Skip to content

chore: bump git-sync to 4.5.0-gke.1 - #2297

Merged
google-oss-prow[bot] merged 2 commits into
GoogleContainerTools:mainfrom
tiffanny29631:gitsync
Sep 22, 2026
Merged

google-oss-prow[bot] merged 2 commits into
GoogleContainerTools:mainfrom
tiffanny29631:gitsync

Conversation

@tiffanny29631

Copy link
Copy Markdown
Contributor

No description provided.

@google-oss-prow
google-oss-prow Bot requested a review from janetkuo September 21, 2026 22:50
@tiffanny29631
tiffanny29631 removed the request for review from janetkuo September 21, 2026 23:07
@Camila-B

Copy link
Copy Markdown
Contributor

/retest

@google-oss-prow google-oss-prow Bot added size/S and removed size/XS labels Sep 22, 2026
@tiffanny29631

Copy link
Copy Markdown
Contributor Author

/retest

@cowsking cowsking left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@google-oss-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: cowsking

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@google-oss-prow
google-oss-prow Bot merged commit 5ace279 into GoogleContainerTools:main Sep 22, 2026
12 checks passed
tiffanny29631 added a commit to tiffanny29631/kpt-config-sync that referenced this pull request Sep 24, 2026
* chore: bump git-sync to 4.5.0-gke.1

* update test predicates for error message
tiffanny29631 added a commit to tiffanny29631/kpt-config-sync that referenced this pull request Sep 24, 2026
* chore: bump git-sync to 4.5.0-gke.1

* update test predicates for error message
google-oss-prow Bot pushed a commit that referenced this pull request Sep 24, 2026
* chore: bump git-sync to 4.5.0-gke.1

* update test predicates for error message
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants