Skip to content

chore: update base image from bookworm to trixie - #2298

Merged
google-oss-prow[bot] merged 1 commit into
GoogleContainerTools:mainfrom
tiffanny29631:vul
Sep 21, 2026
Merged

google-oss-prow[bot] merged 1 commit into
GoogleContainerTools:mainfrom
tiffanny29631:vul

Conversation

@tiffanny29631

Copy link
Copy Markdown
Contributor

No description provided.

@google-oss-prow
google-oss-prow Bot requested a review from janetkuo September 21, 2026 23:01
@tiffanny29631
tiffanny29631 requested review from cowsking and removed request for janetkuo September 21, 2026 23:01

@cowsking cowsking left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@google-oss-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: cowsking

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@google-oss-prow
google-oss-prow Bot merged commit e0ef402 into GoogleContainerTools:main Sep 21, 2026
12 checks passed
tiffanny29631 added a commit to tiffanny29631/kpt-config-sync that referenced this pull request Sep 24, 2026
tiffanny29631 added a commit to tiffanny29631/kpt-config-sync that referenced this pull request Sep 24, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants