Skip to content

Improve log messages - #730

Merged
bluwy merged 2 commits into
mainfrom
improve-log-messages
Sep 4, 2026
Merged

Improve log messages#730
bluwy merged 2 commits into
mainfrom
improve-log-messages

Conversation

@bluwy

@bluwy bluwy commented Aug 26, 2026

Copy link
Copy Markdown
Member

Did a subjective pass and tried to improve/make more consistent of the log messages

Comment thread src/index.ts
: ""
}`,
);
process.exit(result.exitCode);

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This isn't needed as I changed to a throw above, which core.setFailed will make the exit code 1 by default. But while this might be a different exit code than result.exitCode, it doesn't matter in practice as GitHub Actions only has two states: success and failure, which are checked by 0 and non-0 only.

Comment thread src/run.ts
base: branch,
});
core.info(
core.debug(

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I find this info to bloat the logs in practice, so moved this to debug, which can be viewed when the user re-runs the action in debug mode.

tcka4enckoalexei2014-commits

This comment was marked as spam.

tcka4enckoalexei2014-commits

This comment was marked as spam.

tcka4enckoalexei2014-commits

This comment was marked as spam.

@changeset-bot

changeset-bot Bot commented Sep 4, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 72988d2

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@changesets/action Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@beeequeue beeequeue left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

i had a todo item to remove the logging of a big json object from the action, but i cant remember if we did that already

@bluwy

bluwy commented Sep 4, 2026

Copy link
Copy Markdown
Member Author

The big json object is this which I move to core.debug

@beeequeue beeequeue left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

👌

@bluwy
bluwy added this pull request to the merge queue Sep 4, 2026
Merged via the queue into main with commit e08fde7 Sep 4, 2026
7 checks passed
@bluwy
bluwy deleted the improve-log-messages branch September 4, 2026 18:08
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants