Skip to content

Update CODEOWNERS with new Github username - #45

Merged
lforchini merged 1 commit into
nutanix:mainfrom
lforchini:lforchini-codeowners
Sep 25, 2026
Merged

lforchini merged 1 commit into
nutanix:mainfrom
lforchini:lforchini-codeowners

Conversation

@lforchini

Copy link
Copy Markdown
Collaborator

No description provided.

@lforchini
lforchini merged commit a1a71b5 into nutanix:main Sep 25, 2026
3 checks passed
@lforchini
lforchini deleted the lforchini-codeowners branch September 25, 2026 15:15
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants