Conversation
…nfig-managed. (cherry picked from commit 4aec7f7)
(cherry picked from commit 80146d3)
(cherry picked from commit 18ab4ad)
… PRs Replace github.com/openshift/api and github.com/openshift/library-go with github.com/vr4manta/api and github.com/vr4manta/library-go at the heads of openshift/api#3036 and openshift/library-go#2468, needed to build after cherry-picking openshift#442 and openshift#469.
|
Pipeline controller notification For optional jobs, comment This repository is configured in: LGTM mode |
|
@vr4manta: This pull request references SPLAT-2866 which is a valid jira issue. DetailsIn response to this: Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
Important Review skippedAuto reviews are disabled on base/target branches other than the default branch. Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Repository: openshift/coderabbit/.coderabbit.yaml Review profile: CHILL Plan: Enterprise Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
/test all |
|
/pipeline required |
|
Scheduling tests matching the |
|
/retest |
|
@vr4manta: The following test failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
SPLAT-2866
Changes
Enhanced the cluster-cloud-controller-manager-operator (CCCMO) to manage the openshift-config-managed/kube-cloud-config ConfigMap for vSphere platforms, migrating ownership from the Cluster Config Operator (CCO) to CCCMO. This enables CCCMO to:
Notes
This PR is backporting the following PRs in relation to the vSphere multi vCenter day 2 support feature:
Since the original PRs require changes to API and library-go, we need to make sure to update the references here to point to the release-4.21 variants to make sure we do not pull in new functionality.